Skip to content

cargo-deb: 3.7.0 -> 3.8.0 - #558961

Merged
nixpkgs-ci[bot] merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/cargo-deb
Sep 2, 2026
Merged

nixpkgs-ci[bot] merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/cargo-deb

Conversation

@r-ryantm

@r-ryantm r-ryantm commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/kornelski/cargo-deb/releases.

meta.description for cargo-deb is: Cargo subcommand that generates Debian packages from information in Cargo.toml

meta.homepage for cargo-deb is: https://github.com/kornelski/cargo-deb

Updates performed
  • Rust version update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 3.8.0 with grep in /nix/store/43a0qgsq6mb8445r1mi64d2ylbkzq799-cargo-deb-3.8.0
  • found 3.8.0 in filename of file in /nix/store/43a0qgsq6mb8445r1mi64d2ylbkzq799-cargo-deb-3.8.0

Rebuild report (if merged into master) (click to expand)
8 total rebuild path(s)

2 package rebuild(s)

First fifty rebuilds by attrpath

cargo-deb
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/43a0qgsq6mb8445r1mi64d2ylbkzq799-cargo-deb-3.8.0 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A cargo-deb https://github.com/r-ryantm/nixpkgs/archive/b43ff9df2754a1012af1f48347abf415b43086cb.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/b43ff9df2754a1012af1f48347abf415b43086cb#cargo-deb

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/43a0qgsq6mb8445r1mi64d2ylbkzq799-cargo-deb-3.8.0
ls -la /nix/store/43a0qgsq6mb8445r1mi64d2ylbkzq799-cargo-deb-3.8.0/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: b43ff9df2754a1012af1f48347abf415b43086cb


x86_64-linux

✅ 1 package built:
  • cargo-deb

Maintainer pings

cc @matthiasbeyer for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@nixpkgs-ci nixpkgs-ci Bot added the 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". label Sep 2, 2026
@nixpkgs-ci
nixpkgs-ci Bot requested a review from matthiasbeyer September 2, 2026 01:00
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Sep 2, 2026
@matthiasbeyer

Copy link
Copy Markdown
Contributor

@NixOS/nixpkgs-merge-bot merge

@nixpkgs-ci
nixpkgs-ci Bot added this pull request to the merge queue Sep 2, 2026
@nixpkgs-ci

nixpkgs-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

@matthiasbeyer wants to merge this PR.

Requirements to merge this PR with @NixOS/nixpkgs-merge-bot merge:

  • ✅ PR targets a development branch.
  • ✅ PR touches only files of packages in pkgs/by-name/.
  • ✅ PR is at least one of:
  • ✅ PR is not a draft
  • ✅ PR is not blocked by a "changes requested" review from a committer.
  • ✅ matthiasbeyer is a member of @NixOS/nixpkgs-maintainers.
  • ✅ matthiasbeyer is a maintainer of all touched packages on the master branch.

✔️ Queued for merge (#306934)

@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Sep 2, 2026
Merged via the queue into NixOS:master with commit 5426c11 Sep 2, 2026
35 checks passed
@r-ryantm
r-ryantm deleted the auto-update/cargo-deb branch September 2, 2026 12:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants