Skip to content

regclient: 0.11.5 -> 0.11.6 - #559253

Merged
nixpkgs-ci[bot] merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/regclient
Sep 2, 2026
Merged

nixpkgs-ci[bot] merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/regclient

Conversation

@r-ryantm

@r-ryantm r-ryantm commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/regclient/regclient/releases.

meta.description for regclient is: Docker and OCI Registry Client in Go and tooling using those libraries

meta.homepage for regclient is: https://github.com/regclient/regclient

Updates performed
  • Golang update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 0.11.6 in filename of file in /nix/store/8l9vjg9646dxchkdh8fp69rcylk6qdrl-regclient-0.11.6

Rebuild report (if merged into master) (click to expand)
11 total rebuild path(s)

5 package rebuild(s)

First fifty rebuilds by attrpath

regbot
regclient
regctl
regsync
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/8l9vjg9646dxchkdh8fp69rcylk6qdrl-regclient-0.11.6 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A regclient https://github.com/r-ryantm/nixpkgs/archive/839408f5bd5fcbd4f73d887043eda4f4771a577c.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/839408f5bd5fcbd4f73d887043eda4f4771a577c#regclient

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/8l9vjg9646dxchkdh8fp69rcylk6qdrl-regclient-0.11.6
ls -la /nix/store/8l9vjg9646dxchkdh8fp69rcylk6qdrl-regclient-0.11.6/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: 839408f5bd5fcbd4f73d887043eda4f4771a577c


x86_64-linux

✅ 1 package built:
  • regbot (regctl, regsync, regclient)

Maintainer pings

cc @maxbrunet for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@nixpkgs-ci
nixpkgs-ci Bot requested a review from maxbrunet September 2, 2026 19:32
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". labels Sep 2, 2026
@maxbrunet

Copy link
Copy Markdown
Member

@NixOS/nixpkgs-merge-bot merge

@nixpkgs-review-gha

Copy link
Copy Markdown

nixpkgs-review result

Generated using nixpkgs-review-gha (00a1ae2)

Command: nixpkgs-review pr 559253
Commit: 839408f5bd5fcbd4f73d887043eda4f4771a577c (subsequent changes)
Merge: 42ffb7cfc512e6abd44ddeaee14f9c6f5e0f8948

Triggered by @maxbrunet (32458727)
Logs: https://github.com/maxbrunet/nixpkgs-review-gha/actions/runs/33677045021/attempts/1


x86_64-linux

✅ 1 package built:
  • regbot (regclient, regctl, regsync)

aarch64-linux

✅ 1 package built:
  • regbot (regclient, regctl, regsync)

aarch64-darwin (sandbox = false)

✅ 1 package built:
  • regbot (regclient, regctl, regsync)

@nixpkgs-ci
nixpkgs-ci Bot added this pull request to the merge queue Sep 2, 2026
@nixpkgs-ci

nixpkgs-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

@maxbrunet wants to merge this PR.

Requirements to merge this PR with @NixOS/nixpkgs-merge-bot merge:

  • ✅ PR targets a development branch.
  • ✅ PR touches only files of packages in pkgs/by-name/.
  • ✅ PR is at least one of:
  • ✅ PR is not a draft
  • ✅ PR is not blocked by a "changes requested" review from a committer.
  • ✅ maxbrunet is a member of @NixOS/nixpkgs-maintainers.
  • ✅ maxbrunet is a maintainer of all touched packages on the master branch.

✔️ Queued for merge (#306934)

Merged via the queue into NixOS:master with commit 048b32a Sep 2, 2026
30 checks passed
@r-ryantm
r-ryantm deleted the auto-update/regclient branch September 3, 2026 12:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants