Skip to content

sonora: init at 0.30.0 - #559258

Open
Ra77a3l3-jar wants to merge 1 commit into
NixOS:masterfrom
Ra77a3l3-jar:package/sonora
Open

Ra77a3l3-jar wants to merge 1 commit into
NixOS:masterfrom
Ra77a3l3-jar:package/sonora

Conversation

@Ra77a3l3-jar

Copy link
Copy Markdown
Contributor

Init Sonora, a native music streaming client for Spotify, youTube music, and local files

Things done

@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (new) This PR adds a new package 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Sep 2, 2026
Comment thread pkgs/by-name/so/sonora/package.nix Outdated

@c2fc2f c2fc2f Sep 3, 2026

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

On their GitHub, they use these licenses

https://github.com/nolight132/sonora/blob/c4bcaaa4f66c0d3a0563af2b94ae8c0d4b1b209a/flake.nix#L101-L105

Due to fonts and icons. But even their list of licenses is not exhaustive

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

thank you i have missed this. I will fix it right away

@nixpkgs-ci nixpkgs-ci Bot removed the 9.needs: reviewer This PR currently has no reviewers requested and needs attention. label Sep 3, 2026
@c2fc2f

c2fc2f commented Sep 3, 2026

Copy link
Copy Markdown
Member

From what I've seen, the creator uses Nix/NixOS as well, wouldn't it make sense to ask him if he'd like to be a maintainer too?

@Ra77a3l3-jar

Copy link
Copy Markdown
Contributor Author

From what I've seen, the creator uses Nix/NixOS as well, wouldn't it make sense to ask him if he'd like to be a maintainer too?

Yea, i already asked. But im waiting for a response. I will apply all the changes as soon as i get a reply

@Ra77a3l3-jar
Ra77a3l3-jar force-pushed the package/sonora branch 3 times, most recently from c4c290e to c134f68 Compare September 3, 2026 15:55
@Ra77a3l3-jar Ra77a3l3-jar changed the title sonora: init at 0.28.1 sonora: init at 0.29.0 Sep 3, 2026
@Ra77a3l3-jar
Ra77a3l3-jar force-pushed the package/sonora branch 2 times, most recently from 294949a to 311f55b Compare September 4, 2026 18:17
@Ra77a3l3-jar Ra77a3l3-jar changed the title sonora: init at 0.29.0 sonora: init at 0.30.0 Sep 4, 2026
@Ra77a3l3-jar
Ra77a3l3-jar requested a review from c2fc2f September 4, 2026 18:17
@nixpkgs-ci
nixpkgs-ci Bot requested a review from aciceri September 4, 2026 18:38
@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. and removed 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Sep 4, 2026
@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. and removed 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. labels Sep 4, 2026
@nixpkgs-ci nixpkgs-ci Bot added the 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. label Sep 4, 2026
@nixpkgs-ci nixpkgs-ci Bot added the 12.approvals: 1 This PR was reviewed and approved by one person. label Sep 7, 2026
@ZariTen

ZariTen commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 559258
Commit: 488df660490b541893a1d343e0de875908f4fe92


x86_64-linux

✅ 1 package built:
  • sonora

@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 3+ This PR was reviewed and approved by three or more persons. and removed 12.approvals: 1 This PR was reviewed and approved by one person. labels Sep 7, 2026
Comment thread pkgs/by-name/so/sonora/package.nix Outdated
'';

passthru = {
updateScript = ./update.sh;

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
updateScript = ./update.sh;
updateScript = nix-update-script { };

Comment thread pkgs/by-name/so/sonora/update.sh Outdated

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We no longer need this file

Comment thread pkgs/by-name/so/sonora/package.nix
Comment on lines 82 to 92

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

this whole fixupPhase with patchelf and wrapping can be removed by replacing alsa-lib in buildInputs with alsa-lib-with-plugins, and adding autoPatchelfHook to move wayland and vulkan-loader to runtimeDependencies, as seen here (ignore all the darwin-related logic): https://github.com/camuward/sonora/blob/5019c8eeadc252afb433ee0b914eda842895a2c8/flake.nix#L148-L165

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (new) This PR adds a new package 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.approvals: 3+ This PR was reviewed and approved by three or more persons.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants