Skip to content

libblake3: 1.8.6 -> 1.8.7 - #559261

Merged
fgaz merged 1 commit into
NixOS:stagingfrom
r-ryantm:auto-update/libblake3
Sep 3, 2026
Merged

libblake3: 1.8.6 -> 1.8.7#559261
fgaz merged 1 commit into
NixOS:stagingfrom
r-ryantm:auto-update/libblake3

Conversation

@r-ryantm

@r-ryantm r-ryantm commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/BLAKE3-team/BLAKE3/releases.

meta.description for libblake3 is: Official C implementation of BLAKE3

meta.homepage for libblake3 is: https://github.com/BLAKE3-team/BLAKE3/tree/master/c

Updates performed
  • Version update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 1.8.7 with grep in /nix/store/m9q764p4577v7sd6c83abjzlxssn0isr-libblake3-1.8.7
  • found 1.8.7 in filename of file in /nix/store/m9q764p4577v7sd6c83abjzlxssn0isr-libblake3-1.8.7

Rebuild report (if merged into master) (click to expand)
4679 total rebuild path(s)

4673 package rebuild(s)

First fifty rebuilds by attrpath

CuboCore.corepdf
CuboCore.coretime
CuboCore.coretoppings
LycheeSlicer
MMA
OVMFFull
SDL
SDL2
SDL2_Pango
SDL2_gfx
SDL2_image
SDL2_mixer
SDL2_net
SDL2_sound
SDL2_ttf
SDL_Pango
SDL_audiolib
SDL_compat
SDL_gfx
SDL_image
SDL_mixer
SDL_net
SDL_sound
SDL_stretch
SDL_ttf
_1oom
_20kly
_2ship2harkinian
_4d-minesweeper
_64gram
_7kaa
_86Box
_86Box-with-roms
_86box
_86box-with-roms
_90secondportraits
a-keys-path
aab
aaxtomp3
abbaye-des-morts
abracadabra
abuse
actiona
activitywatch
actual-client
adapta-gtk-theme
addwater
adscan
adw-bluetooth
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/m9q764p4577v7sd6c83abjzlxssn0isr-libblake3-1.8.7 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A libblake3 https://github.com/r-ryantm/nixpkgs/archive/2c99b85d13eb4ed8cfcc920e0797bcf071ec55e7.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/2c99b85d13eb4ed8cfcc920e0797bcf071ec55e7#libblake3

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/m9q764p4577v7sd6c83abjzlxssn0isr-libblake3-1.8.7
ls -la /nix/store/m9q764p4577v7sd6c83abjzlxssn0isr-libblake3-1.8.7/bin


Pre-merge build results

Nixpkgs review skipped


Maintainer pings

cc @fgaz @fpletz @silvanshade for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@nixpkgs-ci
nixpkgs-ci Bot requested review from fgaz, fpletz and silvanshade September 2, 2026 19:49
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 501+ This PR causes many rebuilds on Linux and should normally target the staging branches. 10.rebuild-darwin: 101-500 This PR causes between 101 and 500 packages to rebuild on Darwin. 10.rebuild-linux: 2501-5000 This PR causes many rebuilds on Linux and should target the staging branches. 10.rebuild-nixos-tests This PR causes rebuilds for all NixOS tests and should normally target the staging branches. 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". labels Sep 2, 2026
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Sep 3, 2026
@fgaz
fgaz added this pull request to the merge queue Sep 3, 2026
Merged via the queue into NixOS:staging with commit 5bf2cd9 Sep 3, 2026
35 checks passed
@r-ryantm
r-ryantm deleted the auto-update/libblake3 branch September 3, 2026 12:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 101-500 This PR causes between 101 and 500 packages to rebuild on Darwin. 10.rebuild-linux: 501+ This PR causes many rebuilds on Linux and should normally target the staging branches. 10.rebuild-linux: 2501-5000 This PR causes many rebuilds on Linux and should target the staging branches. 10.rebuild-nixos-tests This PR causes rebuilds for all NixOS tests and should normally target the staging branches. 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants