Skip to content

[CAPI@backplane-5.1] Update report - #780

Draft
github-actions[bot] wants to merge 1 commit into
backplane-5.1from
backplane-5.1-pr-patches-capi
Draft

github-actions[bot] wants to merge 1 commit into
backplane-5.1from
backplane-5.1-pr-patches-capi

Conversation

@github-actions

@github-actions github-actions Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

CAPI@backplane-5.1 Update report

  • Updated with today's date

@openshift-ci

openshift-ci Bot commented Sep 9, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: github-actions[bot]
Once this PR has been reviewed and has the lgtm label, please assign bluesort for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants